Conversation
Makefile copied from a similar PR in the OpenWRT repo by @ktgeek See openwrt/packages#26206
Member
|
This package uses |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
First of all, I'm not a C/C++ developer and my knowledge of GCC + Makefile is very rudimentary so apologies for this incomplete PR.
I tried to port an incoming new OpenWRT package named
duperemovesubmitted by @ktgeek and wanted to give it a fair go but it's taking more time than I can affort to so I'll just share my code experiments here.If anyone qualified wants to pick it up, please be my guest! 😅
What I've done:
Dockerfilehadn't changed in a while.configs/x64-3.2.configto.configmake package/symlinks/home/me/Entware/feeds/rtndev/duperemovedidn't get picked up, so I made a symlink myself topackage/feeds/rtndev/duperemoveduperemovefails due to missingbtrfs.hfile. So I tried addingCONFIG_KERNEL_BTRFS_FS=yto the.config, rebuild the toolchain, but it didn't fix it. I then decided to add an include of-I/usr/includetoCFLAGSin the packageMakefile.duperemovewith the defaultMakefilefrom https://github.com/openwrt/packages/pull/26206/commits failed. I believe it's related to the introduction of[[maybe_unused]]attributes and other syntax errors..configfile to use GCC 12 since the default GCC 8.4 didn't support some code syntax from duperemove.CONFIG_BINUTILS_USE_VERSION_2_34=ysince the Entware toolchain wouldn't build with GCC 12 + glibc v2.27Make command:
duperemovestill fails during the linking phase due to the linker not finding some shared libraries likelibiconvorlibintlduperemove/Makefileto expose more include file pathsduperemovebut now it doesn't seem to run properly.I also tried running
duperemoveafter installing the generatedduperemove_0.15.1-1_x64-3.2.ipkpackage manually on my NAS but the same error appeared.At this stage, I've spent two days on this but I'm unfortunately forced to let go due to other priorities. 🥲
See this gist for more info:
Compile tested: intended for x86_64
Run tested: n/a