Repository navigation
Bump github.com/Azure/azure-sdk-for-go/sdk/azcore from 1.23.1 to 1.23.2 in the azure-sdk group - #1074
Merged
Conversation
Bumps the azure-sdk group with 1 update: [github.com/Azure/azure-sdk-for-go/sdk/azcore](https://github.com/Azure/azure-sdk-for-go). Updates `github.com/Azure/azure-sdk-for-go/sdk/azcore` from 1.23.1 to 1.23.2 - [Release notes](https://github.com/Azure/azure-sdk-for-go/releases) - [Commits](Azure/azure-sdk-for-go@sdk/azcore/v1.23.1...sdk/azcore/v1.23.2) --- updated-dependencies: - dependency-name: github.com/Azure/azure-sdk-for-go/sdk/azcore dependency-version: 1.23.2 dependency-type: direct:production update-type: version-update:semver-patch dependency-group: azure-sdk ... Signed-off-by: dependabot[bot] <support@github.com>
dependabot
Bot
requested review from
Dabnsky,
Ferelith-maker,
brayan-trejo,
foodprocessor and
jfantinhardesty
as code owners
October 5, 2026 23:37
jfantinhardesty
had a problem deploying
to
testing
October 7, 2026 16:43 — with
GitHub Actions
Failure
jfantinhardesty
requested a deployment
to
testing
October 7, 2026 16:43 — with
GitHub Actions
In progress
jfantinhardesty
approved these changes
Oct 7, 2026
This branch was successfully deployed
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Bumps the azure-sdk group with 1 update: github.com/Azure/azure-sdk-for-go/sdk/azcore.
Updates
github.com/Azure/azure-sdk-for-go/sdk/azcorefrom 1.23.1 to 1.23.2Release notes
Sourced from github.com/Azure/azure-sdk-for-go/sdk/azcore's releases.
Commits
e6d2415Unmarshalling RFC7231 sets GMT time zone (#27613)27cb570Update dependencies (#27624)14e5800Update CI to the latest two versions of Go (#27623)f73af49Configurations: 'specification/compute/resource-manager/Microsoft.Compute/Bu...339802cAdd repository access guidance to deprecation guide (#27602)a23355a[AutoPR sdk-resourcemanager/compute/armcompute]-generated-from-SDK Generation...1d0aa55Remove feature branches from CI triggers [skip ci] (#27612)a5cb5a0Update storage CODEOWNERS: add Gunjan Singh, remove Sourav Gupta (#27607)31f47a8[AutoPR sdk-resourcemanager/storage/armstorage]-generated-from-SDK Generation...caa10beDeprecate armweightsandbiases (#27560)Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting
@dependabot rebase.Dependabot commands and options
You can trigger Dependabot actions by commenting on this PR:
@dependabot rebasewill rebase this PR@dependabot recreatewill recreate this PR, overwriting any edits that have been made to it@dependabot show <dependency name> ignore conditionswill show all of the ignore conditions of the specified dependency@dependabot ignore <dependency name> major versionwill close this group update PR and stop Dependabot creating any more for the specific dependency's major version (unless you unignore this specific dependency's major version or upgrade to it yourself)@dependabot ignore <dependency name> minor versionwill close this group update PR and stop Dependabot creating any more for the specific dependency's minor version (unless you unignore this specific dependency's minor version or upgrade to it yourself)@dependabot ignore <dependency name>will close this group update PR and stop Dependabot creating any more for the specific dependency (unless you unignore this specific dependency or upgrade to it yourself)@dependabot unignore <dependency name>will remove all of the ignore conditions of the specified dependency@dependabot unignore <dependency name> <ignore condition>will remove the ignore condition of the specified dependency and ignore conditions