Skip to content

auto integration deepin-community/ffmpeg - #4601

Open
deepin-community-ci-bot[bot] wants to merge 1 commit into
masterfrom
auto-integration-35943593554
Open

deepin-community-ci-bot[bot] wants to merge 1 commit into
masterfrom
auto-integration-35943593554

Conversation

@deepin-community-ci-bot

Copy link
Copy Markdown
Contributor

auto integration from pr deepin-community/ffmpeg#43.

Signed-off-by: github-actions <github-actions@users.noreply.github.com>
@deepin-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign goldendeng for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@github-actions

Copy link
Copy Markdown

RepoTagCheck Bot

Listed repo integration are not modified changelog:

repo tagOrsha
deepin-community/ffmpeg c589a75161f70b5c6713a22f24f8c0465de8d546

@github-actions

Copy link
Copy Markdown

INTEGRATION Bot

Integration Details: fix(cve): CVE-2026-75143 - avformat/librist: honor the caller buffer size in librist_read by deepin-ci-robot@deepin-community/ffmpeg by deepin-community-ci-bot[bot]
@deepin-community/ci CVE detected

@deepin-ci-robot

Copy link
Copy Markdown
Contributor

Hi @deepin-community-ci-bot[bot]. Thanks for your PR.

I'm waiting for a deepin-community member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@github-actions

Copy link
Copy Markdown

IntegrationProjector Bot
Deepin Testing Integration Project Manager Info
IssueID: 14047
ProjectItemID: PVTI_lADOABhNec4AL0Jezg8c9TE
IssueAddr: linuxdeepin/developer-center#14047

This branch is waiting to be deployed

1 waiting deployment
TestingIntegrationTesterReview — b0d381ff Waiting Sep 24, 2026 by deepin-community-ci-bot[bot] via pass integration / auto test #7243
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants