Repository navigation
feat: update libxrender to 1:0.9.12-1 - #3
deepin-community-bot[bot] wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
TAG Bot TAG: 1%0.9.12-1 |
|
/integrate |
|
Skipping CI for Draft Pull Request. |
|
/hold |
|
AutoIntegrationPr Bot |
This pull request is requested by @UTsweetyfish.
Basic Information
Old Version:
1:0.9.10-1New Version:
1:0.9.12-1