Skip to content

fix(dev-server): include .vite/* to exclude list. - #363

Merged
yusukebe merged 3 commits into
honojs:mainfrom
vhespanha:main
Jul 11, 2026
Merged

yusukebe merged 3 commits into
honojs:mainfrom
vhespanha:main

Conversation

@vhespanha

Copy link
Copy Markdown
Contributor

The default list of paths to exclude from the dev server does not include .vite/*, this works fine on Node/Bun environments but breaks on Deno.

The default list of paths to exclude from the dev server does not
include `.vite/*`, this works fine on Node/Bun environments but breaks
on Deno.

Signed-off-by: Vinicius Hespanha <mail@vhespanha.dev>
@changeset-bot

changeset-bot Bot commented Jun 9, 2026 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b438ef3

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@hono/vite-dev-server Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@yusukebe

Copy link
Copy Markdown
Member

@vhespanha

Thanks. Can you add a changeset? Run yarn changeset at the top of the project, choose patch, and push the generated changeset file.

@yusukebe
yusukebe merged commit 8af17d4 into honojs:main Jul 11, 2026
4 checks passed
@github-actions github-actions Bot mentioned this pull request Jul 11, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants