Skip to content

feat: CODEOWNERS - #5

Closed
ecv wants to merge 2 commits into
mainfrom
ecv/CODEOWNERS
Closed

ecv wants to merge 2 commits into
mainfrom
ecv/CODEOWNERS

Conversation

@ecv

@ecv ecv commented May 26, 2026

Copy link
Copy Markdown
Contributor

Need this.

@scotwells

Copy link
Copy Markdown
Contributor

GitHub is complaining about an error with this.

image

@ecv

ecv commented May 27, 2026

Copy link
Copy Markdown
Contributor Author

I had the org wrong. But, also, we need these teams to exist in this org before we can merge this.

@ecv
ecv marked this pull request as draft May 27, 2026 17:14
@ecv ecv self-assigned this Aug 6, 2026
@ecv

ecv commented Sep 23, 2026

Copy link
Copy Markdown
Contributor Author

Closing. The team this named does not exist in milo-os, and no existing team owns inventory yet.

@ecv ecv closed this Sep 23, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants