Skip to content

Update CODEOWNERS with shared ownerships - #383

Merged
atol merged 1 commit into
mainfrom
alicelin/update-codeowners
Oct 6, 2026
Merged

atol merged 1 commit into
mainfrom
alicelin/update-codeowners

Conversation

@atol

@atol atol commented Oct 6, 2026

Copy link
Copy Markdown
Contributor

What was changed

Update CODEOWNERS to share more of the ownership of the samples.

Why?

Allow SDK team to manage the overall repository while allowing individual teams to maintain their samples.

@atol
atol requested a review from a team as a code owner October 6, 2026 18:51
@atol
atol merged commit d67795f into main Oct 6, 2026
13 checks passed
@atol
atol deleted the alicelin/update-codeowners branch October 6, 2026 20:06
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants