Skip to content

Add copy-shared-files under shared ownership in CODEOWNERS - #528

Merged
atol merged 1 commit into
mainfrom
alicelin/codeowners-copy-shared-file
Oct 6, 2026
Merged

atol merged 1 commit into
mainfrom
alicelin/codeowners-copy-shared-file

Conversation

@atol

@atol atol commented Oct 6, 2026

Copy link
Copy Markdown
Contributor

What was changed

Update CODEOWNERS so /.scripts/copy-shared-files.mjs is under shared ownership.

Why?

Allow individual teams to add samples without requiring approval from the SDK team.

@atol
atol requested a review from a team as a code owner October 6, 2026 20:41
@atol
atol merged commit 07a4b3a into main Oct 6, 2026
13 checks passed
@atol
atol deleted the alicelin/codeowners-copy-shared-file branch October 6, 2026 20:57
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants