Skip to content

feat: filter tags - #523

Merged
mismosmi merged 1 commit into
mainfrom
filter-tags
Oct 6, 2026
Merged

mismosmi merged 1 commit into
mainfrom
filter-tags

Conversation

@mismosmi

Copy link
Copy Markdown
Collaborator

right now every tag is sent to every client on every revalidation.

This is probably fine for most apps but might not be great in larger, multi-tenant setup or in cases where one might not want to leak a tag (e.G. including numeric ids) to the public internet.

@changeset-bot

changeset-bot Bot commented Sep 25, 2026 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f47e2e5

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 5 packages
Name Type
@toapi/server Minor
@toapi/client Minor
@toapi/common Minor
@toapi/react Minor
@toapi/worker Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@mismosmi
mismosmi changed the base branch from main to debounce-revalidation-stream September 25, 2026 11:02
@mismosmi
mismosmi added this pull request to stack #524 September 25, 2026 11:02
@mismosmi
mismosmi removed this pull request from stack #524 October 5, 2026 22:39
@mismosmi
mismosmi changed the base branch from debounce-revalidation-stream to main October 5, 2026 22:39
@mismosmi
mismosmi merged commit 027a13f into main Oct 6, 2026
1 check passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant