Skip to content

fix: no async transition in useQuery - #530

Merged
mismosmi merged 7 commits into
mainfrom
unified-transition-api
Oct 6, 2026
Merged

mismosmi merged 7 commits into
mainfrom
unified-transition-api

Conversation

@mismosmi

@mismosmi mismosmi commented Oct 5, 2026 •

Copy link
Copy Markdown
Collaborator

Fixes several issues:

  • multiple in-flight revalidations could arrive out of order -> race condition fixed
  • useQuery would run updates as an async transition which blocks react -> await then startTransition instead
  • PubSub.publish would not await callbacks, they were not even async to start with -> revalidated now actually awaits the finished state updates

@changeset-bot

changeset-bot Bot commented Oct 5, 2026 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 48b69f6

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 5 packages
Name Type
@toapi/client Minor
@toapi/common Minor
@toapi/react Minor
@toapi/server Minor
@toapi/worker Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@mismosmi
mismosmi marked this pull request as ready for review October 6, 2026 07:50
@mismosmi
mismosmi requested a review from pex October 6, 2026 07:50
@mismosmi
mismosmi merged commit b108495 into main Oct 6, 2026
1 check passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants